Repository navigation
Conversation
|
Skipping CI for Draft Pull Request. |
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: tosky The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
|
Important
This repository does not receive automatic reviews because it has fewer than 10 stars. ⚙️ Run configuration
Comment |
|
Zuul encountered a syntax error while parsing its while parsing a block mapping |
17fffdd to
48b5f0b
Compare
|
recheck |
Both cinder (generic NFS) and cinder-backup (NFS). Depends-On: openstack-k8s-operators/ci-framework#4233
48b5f0b to
37e71bd
Compare
|
Build failed (check pipeline). Post ✔️ cinder-s2i-content-provider SUCCESS in 16h 36m 46s |
|
recheck |
|
This change depends on a change that failed to merge. Change openstack-k8s-operators/ci-framework#4233 is needed. |
|
recheck |
|
Build failed (check pipeline). Post ✔️ cinder-s2i-content-provider SUCCESS in 16h 25m 38s |
|
recheck |
|
Build failed (check pipeline). Post ✔️ cinder-s2i-content-provider SUCCESS in 4h 47m 08s |
|
recheck |
|
Build failed (check pipeline). Post ✔️ cinder-s2i-content-provider SUCCESS in 2h 16m 08s |
Both cinder (generic NFS) and cinder-backup (NFS).
Depends-On: openstack-k8s-operators/ci-framework#4233