Merge https://github.com/migtools/oadp-cli:oadp-1.6 (ef5fe0e) into oadp-1.6 - #260
Merge https://github.com/migtools/oadp-cli:oadp-1.6 (ef5fe0e) into oadp-1.6#260oadp-rebasebot-app[bot] wants to merge 1 commit into
Conversation
|
Hi @oadp-rebasebot-app[bot]. Thanks for your PR. I'm waiting for a migtools member to verify that this patch is reasonable to test. If it is, they should reply with Regular contributors should join the org to skip this step. Once the patch is verified, the new status will be reflected by the I understand the commands that are listed here. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
56c1717
534edc6 to
56c1717
Compare
|
New changes are detected. LGTM label has been removed. |
56c1717 to
209d925
Compare
059cf67
209d925 to
059cf67
Compare
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: oadp-rebasebot-app[bot], weshayutin The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
3 similar comments
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: oadp-rebasebot-app[bot], weshayutin The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: oadp-rebasebot-app[bot], weshayutin The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: oadp-rebasebot-app[bot], weshayutin The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
|
/retest |
This is an automated rebase PR generated by RebaseBot.
Summary
https://github.com/migtools/oadp-cli:oadp-1.6https://github.com/migtools/oadp-cli:oadp-1.6