Skip to content

Add laozc to OWNERS - #471

Merged
kubernetes-prow[bot] merged 1 commit into
kubernetes-csi:library-developmentfrom
laozc:request-approver-library-development
Sep 7, 2026
Merged

kubernetes-prow[bot] merged 1 commit into
kubernetes-csi:library-developmentfrom
laozc:request-approver-library-development

Conversation

@laozc

@laozc laozc commented Sep 4, 2026

Copy link
Copy Markdown
Member

What type of PR is this?
/kind chore

What this PR does / why we need it:

Which issue(s) this PR fixes:

Fixes #

Special notes for your reviewer:

Does this PR introduce a user-facing change?:

NONE

@kubernetes-prow kubernetes-prow Bot added the release-note-none Denotes a PR that doesn't merit a release note. label Sep 4, 2026
@kubernetes-prow

Copy link
Copy Markdown
Contributor

@laozc: The label(s) kind/chore cannot be applied, because the repository doesn't have them.

Details

In response to this:

What type of PR is this?
/kind chore

What this PR does / why we need it:

Which issue(s) this PR fixes:

Fixes #

Special notes for your reviewer:

Does this PR introduce a user-facing change?:

NONE

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@kubernetes-prow kubernetes-prow Bot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels Sep 4, 2026
@laozc

laozc commented Sep 4, 2026

Copy link
Copy Markdown
Member Author

/approve

@andyzhangx andyzhangx left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm

@kubernetes-prow kubernetes-prow Bot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Sep 7, 2026
@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: andyzhangx, laozc

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow kubernetes-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Sep 7, 2026
@kubernetes-prow
kubernetes-prow Bot merged commit 282e393 into kubernetes-csi:library-development Sep 7, 2026
7 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. release-note-none Denotes a PR that doesn't merit a release note. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants